merge from master - #1
Open
sdmunozsierra wants to merge 351 commits into
Open
Conversation
> Removing tags here will remove them from the "Supported" section on > the Hub readme (and will prevent us from spending cycles rebuilding them > on the official build servers), but the tags will still be available to > users who want them. > See https://github.com/docker-library/official-images#library-definition-files > for more detail on this.
Perform automatic releases to GitHub
Closes #41
See merge request archlinux/archlinux-docker!43
This fixes #42.
In the past, we've observed intermittent network issues while communicating with dockerhub and this should help alleviate that.
Update description on Dockerhub on release Closes #47 See merge request archlinux/archlinux-docker!44
This provides publish:offical with BUILD_COMMIT and enables a normal daily release to archlinux/archlinux on the weekly schedule run.
Run release on PUBLISH_OFFICIAL_LIBRARY as well See merge request archlinux/archlinux-docker!45
Replace publish with multi-destination kaniko See merge request archlinux/archlinux-docker!46
This fixes a circular dependency issue noted in [1]. `curl -s` and removal of `--auto-compress` are needed because of differences in the binaries in alpine. [1] docker-library/official-images#8999 (comment)
Switch to alpine in verify stage See merge request archlinux/archlinux-docker!47
Resolves #45.
The Dockerfiles require them for their sha256sum.
Not needed at runtime and adds non-determinism
The container registry is always going to rewrite parts of the image in an uncontrollable way. As far as we know, it's not possible to download a 1:1 copy of a build output from the container registry (until someone figures this out). As far as I understand it, it also explains why `diffoci --semantic` is a thing and why it's generally considered "good enough" (give current constraints).
Show bit for bit reproducibility of the rootFS
… repro documentation
Add a new `repro` image version / tag providing a bit for bit reproducible image Closes #44 See merge request archlinux/archlinux-docker!96
The publish CI step was mistankely omitted from https://gitlab.archlinux.org/archlinux/archlinux-docker/-/merge_requests/96
Add missing REPRO build arguments in the CI publish step See merge request archlinux/archlinux-docker!98
…h script call in the pre-release CI stage Since https://gitlab.archlinux.org/archlinux/archlinux-docker/-/merge_requests/96, the make-dockerfile.sh script expects the new SOURCE_DATE_EPOCH variable / argument which was mistankely omitted in the pre-release CI stage call to the script.
Add the missing 'source_date_epoch' argument to the make-dockerfile.sh script call in the pre-release CI stage See merge request archlinux/archlinux-docker!99
…cumentation accordingly There's actually a dedicated flag to ignore image name annotations specifically so we don't have to ignore the other attributes with the `--semantic` flag.
Update `diffoci` flag to use `--ignore-image-name` and update user documentation accordingly See merge request archlinux/archlinux-docker!100
We were not patching the correct mirrorlist and therefore the Archive repo snapshot was not honored in the build... Also adding some debug visibility to ease future debug sessions
Fix usage of Archive repo snapshot See merge request archlinux/archlinux-docker!101
We need to skip the `repro` tag since we're unable to adjust required build parameters on the docker-library side. See docker-library/official-images#21327
Expand the repro documentation with missing bits: - The Dockerfile needs to be regenerated with the correct group for title annotation to ensure reproducibility. - The CI_COMMIT_SHA of the original pipeline needs to be honored in the Dockerfile. Also, set the timezome to UTC in Makefile and scripts to ensure consistency in the generated dates / timestamps (e.g. ARCHIVE_SNAPSHOT / SOURCE_DATE_EPOCH), regardless of the timezone of the environment. Otherwise, someone rebuilding the image locally can unexpectedly end up with a different value for those if the system uses a different timezome.
Expand repro documentation and ensure fixed timezome See merge request archlinux/archlinux-docker!102
…rfile template In https://gitlab.archlinux.org/archlinux/archlinux-docker/-/commit/4b15f9a1a11d11474cad201521962dc6192630ea, a placeholder for the ldconfig aux-cache cleanup (required for the repro image) was implemented in the Dockerfile template and was substituted by either `&& rm -f /var/cache/ldconfig/aux-cache` for the repro group or `&& true` for other groups (so that it does nothing). While technically harmless, the resulting `&& true` for the non-repro groups is slightly confusing and may raise some eyebrows (see docker-library/official-images#21366). This change aims to drop the placeholder for the ldconfig aux-cache cleanup in the Dockerfile template and simply expand the "ldconfig + sed" RUN command to include the ldconfig aux-cache cleanup for the repro group. This results in a more precisely targeted substitution without unnecessary and confusing addition in the Dockerfile of the non-repro groups.
Drop placeholder substitution for ldconfig aux-cache cleanup in Dockerfile template See merge request archlinux/archlinux-docker!103
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
No description provided.