Skip to content

Feature/from recommended tests to impacted tests - #208

Merged
RomanDavlyatshin merged 3 commits into
mainfrom
feature/from-recommended-tests-to-impacted-tests
Jun 30, 2026
Merged

RomanDavlyatshin merged 3 commits into
mainfrom
feature/from-recommended-tests-to-impacted-tests

Conversation

@iryabov

@iryabov iryabov commented Jun 19, 2026

Copy link
Copy Markdown
Contributor

No description provided.

@iryabov
iryabov requested a review from RomanDavlyatshin June 19, 2026 06:43
@sonarqubecloud

Copy link
Copy Markdown

@RomanDavlyatshin
RomanDavlyatshin merged commit de1c691 into main Jun 30, 2026
4 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants