Skip to content

Version Packages - #52

Merged
Prashant-7718 merged 1 commit into
mainfrom
changeset-release/main
Sep 17, 2026
Merged

Prashant-7718 merged 1 commit into
mainfrom
changeset-release/main

Conversation

@github-actions

Copy link
Copy Markdown
Contributor

This PR was opened by the Changesets release GitHub action. When you're ready to do a release, you can merge this and the packages will be published to npm automatically. If you're not ready to do a release yet, that's fine, whenever you add more changesets to main, this PR will be updated.

Releases

@mermaidchart/sdk@0.2.8

Patch Changes

@vercel

vercel Bot commented Sep 17, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
office-plugin-site Ignored Ignored Sep 17, 2026 3:38pm UTC

Request Review

@Prashant-7718
Prashant-7718 merged commit e6a79fc into main Sep 17, 2026
28 of 50 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant