Skip to content

Update CODEOWNERS - #883

Merged
quetzalliwrites merged 1 commit into
mainfrom
codeowners-auto-updates-docs
Aug 24, 2026
Merged

Update CODEOWNERS#883
quetzalliwrites merged 1 commit into
mainfrom
codeowners-auto-updates-docs

Conversation

@localstack-bot

Copy link
Copy Markdown
Collaborator

Automatic update of CODEOWNERS file.

@cloudflare-workers-and-pages

Copy link
Copy Markdown

Deploying localstack-docs with  Cloudflare Pages  Cloudflare Pages

Latest commit: 844daa6
Status: ✅  Deploy successful!
Preview URL: https://72412150.localstack-docs.pages.dev
Branch Preview URL: https://codeowners-auto-updates-docs.localstack-docs.pages.dev

View logs

@quetzalliwrites
quetzalliwrites merged commit 8d13929 into main Aug 24, 2026
1 of 2 checks passed
@quetzalliwrites
quetzalliwrites deleted the codeowners-auto-updates-docs branch August 24, 2026 11:51
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants