Skip to content

chore: update version to 5.1.1 - #36

Merged
jhoward1994 merged 1 commit into
mainfrom
releases/5.1.1
Dec 19, 2025
Merged

jhoward1994 merged 1 commit into
mainfrom
releases/5.1.1

Conversation

@jhoward1994

Copy link
Copy Markdown
Contributor

No description provided.

@changeset-bot

changeset-bot Bot commented Dec 19, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f34e6d6

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@sonarqubecloud

Copy link
Copy Markdown

@jhoward1994 jhoward1994 self-assigned this Dec 19, 2025
@jhoward1994

Copy link
Copy Markdown
Contributor Author

I'll release on npm from here (+ tag on github)

as per #30

@jhoward1994

Copy link
Copy Markdown
Contributor Author

@jhoward1994
jhoward1994 merged commit 33dcb4f into main Dec 19, 2025
5 checks passed
@jhoward1994
jhoward1994 deleted the releases/5.1.1 branch December 19, 2025 11:40
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants