chore(deps): bump k8s.io/client-go from 0.35.7 to 0.35.8 - #395
Conversation
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #395 +/- ##
=======================================
Coverage 50.28% 50.28%
=======================================
Files 19 19
Lines 2118 2118
=======================================
Hits 1065 1065
Misses 921 921
Partials 132 132
Flags with carried forward coverage won't be shown. Click here to find out more. ☔ View full report in Codecov by Harness. 🚀 New features to boost your workflow:
|
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: infernus01 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
|
@dependabot[bot]: rebase DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. |
648db5d to
bc54c7f
Compare
Bumps [k8s.io/client-go](https://github.com/kubernetes/client-go) from 0.35.7 to 0.35.8. - [Changelog](https://github.com/kubernetes/client-go/blob/master/CHANGELOG.md) - [Commits](kubernetes/client-go@v0.35.7...v0.35.8) --- updated-dependencies: - dependency-name: k8s.io/client-go dependency-version: 0.35.8 dependency-type: direct:production update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
bc54c7f to
390eef5
Compare
Bumps k8s.io/client-go from 0.35.7 to 0.35.8.
Commits
3614e2dUpdate dependencies to v0.35.8 tag7bd3cb5Merge pull request #141225 from liggitt/golang-deps-1-350fab4f6Update golang.org/x deps