Skip to content

fix: debounce revalidation stream - #520

Merged
mismosmi merged 3 commits into
mainfrom
debounce-revalidation-stream
Sep 25, 2026
Merged

mismosmi merged 3 commits into
mainfrom
debounce-revalidation-stream

Conversation

@mismosmi

Copy link
Copy Markdown
Collaborator

revalidation stream can cause many updates in fast succession and submit the same tags many times over a short period.

debounce and dedupe the stream per consumer.

@changeset-bot

changeset-bot Bot commented Sep 24, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7f65018

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@toapi/server Patch
@toapi/client Patch
@toapi/common Patch
@toapi/react Patch
@toapi/worker Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mismosmi
mismosmi added this pull request to stack #524 September 25, 2026 11:02
@mismosmi
mismosmi force-pushed the debounce-revalidation-stream branch from ec567a0 to 7f65018 Compare September 25, 2026 13:03
@mismosmi
mismosmi merged commit 8aca1e8 into main Sep 25, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant