Skip to content

fix: force invalidation immediately on mutation or manual revalidation - #521

Merged
mismosmi merged 3 commits into
mainfrom
force-invalidation-immediately
Sep 25, 2026
Merged

mismosmi merged 3 commits into
mainfrom
force-invalidation-immediately

Conversation

@mismosmi

Copy link
Copy Markdown
Collaborator

Resolves #519

@mismosmi
mismosmi requested a review from pex September 25, 2026 09:20
@changeset-bot

changeset-bot Bot commented Sep 25, 2026 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4093d6d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@toapi/client Patch
@toapi/common Patch
@toapi/react Patch
@toapi/server Patch
@toapi/worker Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mismosmi
mismosmi merged commit ff532bd into main Sep 25, 2026
1 check passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

mutation.revalidated resolves before a debounced refetch

1 participant